Skip to content

Update to s-m-d/v7, go 1.27 - #636

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
liggitt:smdv7
Sep 4, 2026
Merged

Update to s-m-d/v7, go 1.27#636
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
liggitt:smdv7

Conversation

@liggitt

@liggitt liggitt commented Sep 4, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow kubernetes-prow Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 4, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from Jefftree and jpbetz September 4, 2026 16:39
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Sep 4, 2026
@liggitt liggitt changed the title WIP: Update to s-m-d/v7, go 1.27 Update to s-m-d/v7, go 1.27 Sep 4, 2026
@kubernetes-prow kubernetes-prow Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 4, 2026

@jpbetz jpbetz left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 4, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jpbetz, liggitt

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 4, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 9ab3195 into kubernetes:master Sep 4, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants