This repository was archived by the owner on Aug 5, 2026. It is now read-only.
Feat/838 switch from Railway to Render - #901
Merged
lightwalker-eth merged 20 commits intoAug 14, 2025
Merged
Conversation
🦋 Changeset detectedLatest commit: e04ed1f The changes in this PR will be included in the next version bump. This PR includes changesets to release 11 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
…railway-to-a-new-hosting-provider
lightwalker-eth
left a comment
Member
There was a problem hiding this comment.
@BanaSeba Hey thanks for your updates 👍 Reviewed and shared feedback
lightwalker-eth
left a comment
Member
There was a problem hiding this comment.
@BanaSeba Thanks for updates 👍
…railway-to-a-new-hosting-provider
lightwalker-eth
approved these changes
Aug 14, 2025
lightwalker-eth
left a comment
Member
There was a problem hiding this comment.
@BanaSeba Looks good 👍
lightwalker-eth
deleted the
feat/838-switch-from-railway-to-a-new-hosting-provider
branch
August 14, 2025 16:27
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to subscribe to this conversation on GitHub.
Already have an account?
Sign in.
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Removing Railway Terraform and putting Render instead. This affects only the "test" environment, not production.
Related to: #838