Skip to content

feat: adding a footer to templates + link components and styles utilized - #3319

Merged
martascorreia merged 15 commits into
mainfrom
feat/add-template-links
Sep 10, 2026
Merged

martascorreia merged 15 commits into
mainfrom
feat/add-template-links

Conversation

@martascorreia

@martascorreia martascorreia commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

Description:

closes #1554

Also added some further documentation for this ticket #3285

Definition of Reviewable:

  • Documentation is created/updated
  • Migration Guide is created/updated
  • E2E tests (features, a11y, bug fixes) are created/updated
  • Stories (features, a11y) are created/updated
  • relevant tickets are linked

@changeset-bot

changeset-bot Bot commented Sep 7, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b5f2897

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@solid-design-system/docs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

🚀 Storybook has been deployed for branch feat_add-template-links

@martascorreia martascorreia moved this from 📋 Backlog to 🏗 In progress in Solid Design System Project Board Sep 8, 2026
balco0110
balco0110 previously approved these changes Sep 9, 2026
Comment thread packages/docs/CONTRIBUTING.md Outdated
@martascorreia
martascorreia merged commit e08aef1 into main Sep 10, 2026
19 of 20 checks passed
@github-project-automation github-project-automation Bot moved this from 👀 In review to ✅ Done in Solid Design System Project Board Sep 10, 2026
@martascorreia
martascorreia deleted the feat/add-template-links branch September 10, 2026 14:38

This branch was previously deployed

1 inactive deployment
github-pages b5f28974 Deployed Sep 10, 2026 by martascorreia via deploy #5422
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: ✅ Done

Development

Successfully merging this pull request may close these issues.

docs[dev]: ✨ add link from templates to components and styles

7 participants