Skip to content

feat: added the dropdown with menu items story - #3322

Merged
rahulsuvarna18 merged 11 commits into
mainfrom
feat/dropdown-menu-template
Sep 15, 2026
Merged

rahulsuvarna18 merged 11 commits into
mainfrom
feat/dropdown-menu-template

Conversation

@rahulsuvarna18

@rahulsuvarna18 rahulsuvarna18 commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Description:

closes #3018

Definition of Reviewable:

  • Stories (features, a11y) are created/updated
  • relevant tickets are linked

@changeset-bot

changeset-bot Bot commented Sep 8, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 014cb6c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@solid-design-system/docs Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

🚀 Storybook has been deployed for branch feat_dropdown-menu-template

balco0110
balco0110 previously approved these changes Sep 8, 2026

@balco0110 balco0110 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@martascorreia martascorreia left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

There is an empty Default story in the page and right side menu

@rahulsuvarna18
rahulsuvarna18 marked this pull request as draft September 9, 2026 13:16
Comment thread packages/docs/src/stories/templates/dropdown.stories.ts Outdated
Comment thread packages/docs/src/stories/templates/dropdown.stories.ts Outdated
@rahulsuvarna18
rahulsuvarna18 marked this pull request as ready for review September 10, 2026 14:31
martascorreia
martascorreia previously approved these changes Sep 10, 2026
Comment thread .changeset/cool-cats-search.md Outdated
@rahulsuvarna18
rahulsuvarna18 merged commit 9b3f8a3 into main Sep 15, 2026
20 checks passed
@rahulsuvarna18
rahulsuvarna18 deleted the feat/dropdown-menu-template branch September 15, 2026 09:16
@github-project-automation github-project-automation Bot moved this from 👀 In review to ✅ Done in Solid Design System Project Board Sep 15, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: ✅ Done

Development

Successfully merging this pull request may close these issues.

feat[dev]: implement dropdown menu template

6 participants