Skip to content

fix: Menu dropdown closing on clicks outside the component - #3336

Open
rahulsuvarna18 wants to merge 4 commits into
mainfrom
fix/menu-dropdown-close
Open

rahulsuvarna18 wants to merge 4 commits into
mainfrom
fix/menu-dropdown-close

Conversation

@rahulsuvarna18

@rahulsuvarna18 rahulsuvarna18 commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Description:

closes #3334

Definition of Reviewable:

  • Stories (features, a11y) are created/updated
  • relevant tickets are linked

@changeset-bot

changeset-bot Bot commented Sep 14, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e6eda58

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@solid-design-system/docs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

🚀 Storybook has been deployed for branch fix_menu-dropdown-close

@MartaPintoTeixeira MartaPintoTeixeira removed their assignment Sep 17, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: 👀 In review

Development

Successfully merging this pull request may close these issues.

fix: the menu template dropdown closes every downtown on click

6 participants